effective-creation-timeout & new metrics for use by dependency-watchdog - #1104
Conversation
Test Log-1Testing done with virtual provider. Real provider test logs will be attached to PR's raised for the specific providers. Effective Machine Creation Timeout1. Check MCD2. Add effective-creation-timeout annotation to MCD2. Check annotation propagated to MachineSet2. Increase replicas & check annotation propagated to new Machine object3. Case: Creation Success MetricsMachines should be created and new Node joins the cluster. From local prometheus http://127.0.0.1:9090/prometheus 4. Case: Machine Join Failure
Machine object From Prometheus |
|
/assign |
Co-authored-by: Aaron Francis Fernandes <79958509+aaronfern@users.noreply.github.com>
Co-authored-by: Aaron Francis Fernandes <79958509+aaronfern@users.noreply.github.com>
aaronfern
left a comment
There was a problem hiding this comment.
Just a few minor commits from me. In general lgtm once the comments are resolved.
Co-authored-by: Aaron Francis Fernandes <79958509+aaronfern@users.noreply.github.com>
Co-authored-by: Aaron Francis Fernandes <79958509+aaronfern@users.noreply.github.com>
aaronfern
left a comment
There was a problem hiding this comment.
Thanks for the changes!
/lgtm
|
LGTM label has been added. DetailsGit tree hash: 6452cfc88d126af93c08aa83b082cf38be970903 |
|
/approve |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: aaronfern The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
What this PR does / why we need it:
Which issue(s) this PR fixes:
Fixes #1103
Fixes #1098
Special notes for your reviewer:
Release note: